Fix TRAX kernel Makefile - #39
Open
ChaseParate wants to merge 1 commit into
Open
ChaseParate wants to merge 1 commit into
ChaseParate wants to merge 1 commit into
Conversation
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fixes a few issues:
.PHONYto avoid make thinking the targets were already built.riscvtarget has the same name as theriscvdirectory, so make would think the target was already built, as the directory already exists (and thus not rebuild the kernel if necessary)mkdircalls into the Makefile, but this can also be achieved by adding a.gitkeepfile or something of the sort in that directory. I'm open to either.