Skip to content

Bump software.amazon.awssdk:bom from 2.16.99 to 2.17.79 in /app - #206

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/gradle/app/software.amazon.awssdk-bom-2.17.79
Closed

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/gradle/app/software.amazon.awssdk-bom-2.17.79

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Nov 10, 2021

Copy link
Copy Markdown
Contributor

Bumps software.amazon.awssdk:bom from 2.16.99 to 2.17.79.

Changelog

Sourced from software.amazon.awssdk:bom's changelog.

2.17.79 2021-11-10

AWS Backup

  • Features

    • AWS Backup SDK provides new options when scheduling backups: select supported services and resources that are assigned to a particular tag, linked to a combination of tags, or can be identified by a partial tag value, and exclude resources from their assignments.

AWS Resilience Hub

  • Features

    • Initial release of AWS Resilience Hub, a managed service that enables you to define, validate, and track the resilience of your applications on AWS

AWS SDK for Java

  • Features

    • Added a new dualstackEnabled property to every client builder, which can be used to make calls be invoked against AWS endpoints which return IPv6 records. This can also be enabled via the AWS_USE_DUALSTACK_ENDPOINT environment variable, aws.useDualstackEndpoint system property, or the use_dualstack_endpoint profile file property.

AWS SDK for Java v2

  • Features

    • Added a new fipsEnabled property to every client builder, which can be used to make calls be invoked against AWS endpoints which are FIPS compliant. This can also be enabled via the AWS_USE_FIPS_ENDPOINT environment variable, aws.useFipsEndpoint system property, or the use_fips_endpoint profile file property.

Amazon EC2 Container Service

  • Features

    • This release adds support for container instance health.

Amazon S3

  • Deprecations

    • Deprecated S3Configuration.Builder's dualstackEnabled in favor of the new service-standard S3ClientBuilder.dualstackEnabled.

Amazon S3 Control

  • Deprecations

    • Deprecated S3ControlConfiguration.Builder's dualstackEnabled in favor of the new service-standard S3ControlClientBuilder.dualstackEnabled.
    • Deprecated S3ControlConfiguration.Builder's fipsModeEnabled in favor of the new service-standard S3ControlClientBuilder.fipsEnabled.

Apache HTTP Client

  • Features

    • Add DNS resolver override support for Apache HTTP Client

2.17.78 2021-11-09

AWS Batch

  • Features

    • Adds support for scheduling policy APIs.

AWS Health APIs and Notifications

  • Features

    • Documentation updates for AWS Health.

AWS IoT Greengrass V2

  • Features

    • This release adds support for Greengrass core devices running Windows. You can now specify name of a Windows user to run a component.

AWS SDK for Java v2

  • Features

    • Updated service endpoint metadata.

... (truncated)

Commits
  • 28991ff Temporarily disable TranscribeStreaming integration tests (#2840)
  • 3d475c1 Merge pull request #1832 from aws/staging/3584f3ce-f2aa-4660-b709-48ced3fad9c5
  • de6663e Release 2.17.79. Updated CHANGELOG.md, README.md and all pom.xml.
  • 707088f AWS Resilience Hub Update: Initial release of AWS Resilience Hub, a managed s...
  • a419f5a Amazon EC2 Container Service Update: This release adds support for container ...
  • f66b0c9 AWS Backup Update: AWS Backup SDK provides new options when scheduling backup...
  • f8a440c adds DNS resolver configuration option for ApacheHttpClient (#2834)
  • 821a342 Added UploadDirectoryOverrideConfiguration.Builder#uploadFileRequestTransform...
  • 4656690 Release endpoint variants: dualstack and fips for all services. (#2830)
  • 3b6ab48 Change CODEOWNERS to be team-based
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [software.amazon.awssdk:bom](https://github.com/aws/aws-sdk-java-v2) from 2.16.99 to 2.17.79.
- [Release notes](https://github.com/aws/aws-sdk-java-v2/releases)
- [Changelog](https://github.com/aws/aws-sdk-java-v2/blob/master/CHANGELOG.md)
- [Commits](aws/aws-sdk-java-v2@2.16.99...2.17.79)

---
updated-dependencies:
- dependency-name: software.amazon.awssdk:bom
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file java Pull requests that update Java code labels Nov 10, 2021
@dependabot @github

dependabot Bot commented on behalf of github Nov 11, 2021

Copy link
Copy Markdown
Contributor Author

Superseded by #207.

@dependabot dependabot Bot closed this Nov 11, 2021
@dependabot
dependabot Bot deleted the dependabot/gradle/app/software.amazon.awssdk-bom-2.17.79 branch November 11, 2021 22:06
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file java Pull requests that update Java code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants